Data track extension ID collision: intermediate step. - #1825
Merged
Merged
Conversation
Unfortunately, the extension ID allocated for participant SID across relay got used by data packet definition too. To address that, the plan - Allocate internally needed participant SID at 255 (highest extension ID) - Accept both on read side, try HIGH first and then the original - Clean up code after a full deploy - Rename LOW to E2EE extension ID.
🦋 Changeset detectedLatest commit: e2a8163 The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Added a note about data track extension ID collision as an intermediate step.
milos-lk
approved these changes
Sep 29, 2026
boks1971
commented
Sep 29, 2026
| // forward metrics | ||
| uint32 forward_latency = 48; | ||
| uint32 forward_jitter = 49; | ||
| uint32 forward_jitter = 49 [deprecated = true]; |
Contributor
Author
There was a problem hiding this comment.
am deprecating this from the result of a different effort. Forwarding latency jitter is not used anywhere. So, marking it unused for now.
This was referenced Sep 29, 2026
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Unfortunately, the extension ID allocated for participant SID across relay got used by data packet definition too. To address that, the plan